Skip to content

Bump master branch to 1.11.0-dev - #1387

Merged
yuumasato merged 2 commits into
ComplianceAsCode:masterfrom
yuumasato:bump-to-1.11-dev
Sep 9, 2026
Merged

Bump master branch to 1.11.0-dev#1387
yuumasato merged 2 commits into
ComplianceAsCode:masterfrom
yuumasato:bump-to-1.11-dev

Conversation

@yuumasato

Copy link
Copy Markdown
Member

No description provided.

When going from version-dev to version, just remove the prefix
Branch release-1.10 was created.

@Vincent056 Vincent056 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Vincent056, yuumasato

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [Vincent056,yuumasato]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@github-actions

github-actions Bot commented Sep 4, 2026

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1387-3036d7163c88cb19185dafb5982be8d4a9bb0262

@yuumasato yuumasato added this to the 1.11.0 milestone Sep 9, 2026
@yuumasato
yuumasato merged commit cd2c2bd into ComplianceAsCode:master Sep 9, 2026
23 of 27 checks passed
@yuumasato
yuumasato deleted the bump-to-1.11-dev branch September 9, 2026 08:23
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants