Skip to content

Bump config manifest CSV version name to 1.10.0 dev - #1146

Closed
yuumasato wants to merge 1 commit into
ComplianceAsCode:masterfrom
yuumasato:bump-version-1.10.0-dev
Closed

yuumasato wants to merge 1 commit into
ComplianceAsCode:masterfrom
yuumasato:bump-version-1.10.0-dev

Conversation

@yuumasato

Copy link
Copy Markdown
Member

Bump version in master to v1.10.0-dev after branching release-1.9.

@openshift-ci
openshift-ci Bot requested review from mrogers950 and xiaojiey April 2, 2026 23:07
@openshift-ci openshift-ci Bot added the approved label Apr 2, 2026
@yuumasato

Copy link
Copy Markdown
Member Author

/test e2e-aws-parallel-arm

@yuumasato
yuumasato force-pushed the bump-version-1.10.0-dev branch from bfe2925 to b971bd7 Compare April 14, 2026 22:12
@yuumasato
yuumasato force-pushed the bump-version-1.10.0-dev branch 2 times, most recently from 97e7fec to 5af5f6d Compare June 23, 2026 09:26
@github-actions

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-97e7fec58671ed8d5d5187d0bfc4007807cf5347

@github-actions

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-5af5f6d42d2de7e133828bb8dcdc424f73679ed4

Comment thread cmd/manager/cel_scanner_test.go
Comment thread cmd/manager/cel_scanner_test.go
Comment thread images/must-gather/Containerfile
Comment thread pkg/apis/compliance/v1alpha1/rule_scanner_test.go
Comment thread pkg/apis/compliance/v1alpha1/rule_types.go Outdated
Comment thread pkg/controller/compliancescan/aggregator.go
Comment thread pkg/controller/compliancescan/compliancescan_controller_test.go Outdated
Comment thread pkg/controller/compliancescan/scan.go
@Vincent056

Copy link
Copy Markdown

I think the PR looks good, except for some unwanted formatting changes.

@yuumasato

yuumasato commented Jun 24, 2026 •

Copy link
Copy Markdown
Member Author

The indentations were added by make fmt, I can remove them.
Eventually they will be added again by somebody.

@yuumasato
yuumasato force-pushed the bump-version-1.10.0-dev branch from 5af5f6d to 8ebd888 Compare June 24, 2026 12:44
@github-actions

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-8ebd88877ceba8d61a19f232c1e0a0376c48a0e4

@yuumasato
yuumasato force-pushed the bump-version-1.10.0-dev branch from 8ebd888 to 8a43dad Compare July 2, 2026 12:39
@github-actions

github-actions Bot commented Jul 2, 2026

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-8a43dad14fbb59bae67ded206fbb4cb117a944e7

@yuumasato

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci

openshift-ci Bot commented Aug 7, 2026

Copy link
Copy Markdown

Stale PRs are closed after 21d of inactivity.

If this PR is still relevant, comment to refresh it or remove the stale label.
Mark the PR as fresh by commenting /remove-lifecycle stale.

If this PR is safe to close now please do so with /close.

/lifecycle stale

@github-actions

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-ea3b4a1f07af5b6b37587a4144060fc5a1118da7

@yuumasato yuumasato added this to the 1.10.0 milestone Aug 11, 2026
taimurhafeez
taimurhafeez previously approved these changes Aug 11, 2026
@taimurhafeez

Copy link
Copy Markdown
Collaborator

/lgtm

@yuumasato

yuumasato commented Aug 20, 2026 •

Copy link
Copy Markdown
Member Author

Rebased and updated to fix one leftover version from #1351

taimurhafeez
taimurhafeez previously approved these changes Aug 20, 2026
@openshift-ci openshift-ci Bot added the lgtm label Aug 20, 2026
@taimurhafeez

Copy link
Copy Markdown
Collaborator

/lgtm

@openshift-ci

openshift-ci Bot commented Aug 20, 2026 •

Copy link
Copy Markdown

@yuumasato: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-config 5af5f6d link true /test e2e-aws-config
ci/prow/e2e-rosa 8ebd888 link true /test e2e-rosa
ci/prow/e2e-aws-parallel-arm 8ebd888 link true /test e2e-aws-parallel-arm
ci/prow/e2e-aws-serial-arm 8ebd888 link true /test e2e-aws-serial-arm
ci/prow/e2e-aws-deployment 8ebd888 link true /test e2e-aws-deployment
ci/prow/e2e-aws-scan-config 8ebd888 link true /test e2e-aws-scan-config

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@github-actions

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-4651754390d6a3197d872f883dd6e4468db0c392

@yuumasato
yuumasato force-pushed the bump-version-1.10.0-dev branch from 4651754 to d58f124 Compare August 24, 2026 14:51
@github-actions

Copy link
Copy Markdown

🤖 To deploy this PR, run the following command:

make catalog-deploy CATALOG_IMG=ghcr.io/complianceascode/compliance-operator-catalog:1146-d58f12437e3a109ea93f1aeb50ddaca70eb394ac

@yuumasato yuumasato changed the title Bump version 1.10.0 dev Bump config manifest CSV version name to 1.10.0 dev Aug 31, 2026

@Vincent056 Vincent056 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Vincent056, yuumasato

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [Vincent056,yuumasato]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@yuumasato

Copy link
Copy Markdown
Member Author

Folded into #1387

@yuumasato yuumasato closed this Sep 4, 2026
@yuumasato
yuumasato deleted the bump-version-1.10.0-dev branch September 4, 2026 21:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants