Bump go.podman.io/image/v5 from 5.41.1 to 5.41.2 - #528
dependabot[bot] wants to merge 1 commit into
Conversation
Bumps [go.podman.io/image/v5](https://github.com/podman-container-tools/container-libs) from 5.41.1 to 5.41.2. - [Release notes](https://github.com/podman-container-tools/container-libs/releases) - [Commits](podman-container-tools/container-libs@image/v5.41.1...image/v5.41.2) --- updated-dependencies: - dependency-name: go.podman.io/image/v5 dependency-version: 5.41.2 dependency-type: direct:production update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: rashmigottipati The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Bumps go.podman.io/image/v5 from 5.41.1 to 5.41.2.
Commits
b5becb3bump image to 5.41.232b2c7ego.mod: require go.podman.io/storage v1.64.16eac452bump storage to 1.64.13b75133Merge commit from fork8049716Merge branch 'LOCAL-oci-archive-6.1' into LOCAL-changes-security-6.1b3122a5Merge branch 'LOCAL-storage-extract-archive-6.1' into LOCAL-changes-security-6.15ec1386Merge branch 'LOCAL-storage-create-archive-6.1' into LOCAL-changes-security-6.12de0f0bMerge branch 'LOCAL-storage-tests-6.1' into LOCAL-changes-security-6.1dc134cbPrevent symbolic link escapes when consuming oci-archive: archivesd06a599Turn tempDirOCIRef into stateful unpackedArchiveDependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)