Skip to content

docs(data-collection): Update cookie filtering examples - #19700

Open
s1gr1d wants to merge 1 commit into
masterfrom
sig/cookie-header-docs
Open

s1gr1d wants to merge 1 commit into
masterfrom
sig/cookie-header-docs

Conversation

@s1gr1d

@s1gr1d s1gr1d commented Sep 29, 2026

Copy link
Copy Markdown
Member

Update the examples based on the changes in those PRs:

Based on Sentry Conventions, we don't collect cookies within their own keyed attribtue anymore.

@vercel

vercel Bot commented Sep 29, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
develop-docs Ready Ready Preview Sep 29, 2026 1:51pm UTC
1 Skipped Deployment
Project Deployment Actions Updated
sentry-docs Ignored Ignored Sep 29, 2026 1:51pm UTC

Request Review

@github-actions github-actions Bot added the sdk-develop-docs PRs touching develop-docs/sdk label Sep 29, 2026
@s1gr1d
s1gr1d requested review from Lms24 and cleptric September 29, 2026 13:50
@github-actions github-actions Bot added the Priority: Needs Triage The PR description is missing valid priority information label Sep 29, 2026
@github-actions

Copy link
Copy Markdown
Contributor

Please update the IS YOUR CHANGE URGENT? section of the PR description:

  • Select exactly one option.
  • If you selected an urgent or other deadline, provide the date as YYYY-MM-DD.

This information helps the Docs team prioritize your review.

This branch was successfully deployed

1 active deployment
Preview – develop-docs — 8864173b Deployed Sep 29, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Priority: Needs Triage The PR description is missing valid priority information sdk-develop-docs PRs touching develop-docs/sdk

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant