Skip to content

fix(performance settings): Updated page to direct link to settings, and add clearer screenshot - #19589

Merged
sfanahata merged 1 commit into
masterfrom
sanahata/fix/update-performance-config-instructions
Sep 23, 2026
Merged

sfanahata merged 1 commit into
masterfrom
sanahata/fix/update-performance-config-instructions

Conversation

@sfanahata

@sfanahata sfanahata commented Sep 23, 2026 •

Copy link
Copy Markdown
Contributor

DESCRIBE YOUR PR

Updated page to direct link to settings, and add clearer screenshot

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.
Select exactly one option. For deadlines, replace YYYY-MM-DD with the due date. You can update this information later by editing the PR description.

  • Urgent deadline (GA date, etc.): YYYY-MM-DD
  • Other deadline: 2026-09-23
  • No deadline: Not urgent, can wait up to 1 week+

SLA

  • Teamwork makes the dream work, so please add a reviewer to your PRs.
  • Please give the docs team up to 1 week to review your PR unless you've supplied a deadline.

Thanks in advance for your help!

PRE-MERGE CHECKLIST

Make sure you've checked the following before merging your changes:

  • Checked Vercel preview for correctness, including links
  • PR was reviewed and approved by any necessary SMEs (subject matter experts)
  • PR was reviewed and approved by a member of the Sentry docs team

@vercel

vercel Bot commented Sep 23, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
sentry-docs Ready Ready Preview Sep 23, 2026 7:48pm UTC
1 Skipped Deployment
Project Deployment Actions Updated
develop-docs Ignored Ignored Sep 23, 2026 7:48pm UTC

Request Review

@codeowner-assignment
codeowner-assignment Bot requested a review from a team September 23, 2026 19:41
@github-actions github-actions Bot added the Priority: Needs Triage The PR description is missing valid priority information label Sep 23, 2026
@github-actions github-actions Bot added Priority: Deadline Docs review has a specific non-urgent due date and removed Priority: Needs Triage The PR description is missing valid priority information labels Sep 23, 2026

@coolguyzone coolguyzone left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nice!

@sfanahata
sfanahata merged commit 8b20678 into master Sep 23, 2026
25 checks passed
@sfanahata
sfanahata deleted the sanahata/fix/update-performance-config-instructions branch September 23, 2026 22:49

This branch was successfully deployed

1 active deployment
Preview – sentry-docs — 949ca29c Deployed Sep 23, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Priority: Deadline Docs review has a specific non-urgent due date

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants