Skip to content

Add maintainers section to CONTRIBUTING.md - #521

Open
notmandatory wants to merge 1 commit into
bitcoindevkit:masterfrom
notmandatory:docs/update_contributing
Open

notmandatory wants to merge 1 commit into
bitcoindevkit:masterfrom
notmandatory:docs/update_contributing

Conversation

@notmandatory

@notmandatory notmandatory commented Jul 29, 2026

Copy link
Copy Markdown
Member

Description

I borrowed from the rust-bitcoin CONTRIBUTING.md file to add a maintainers section to our CONTRIBUTING.md file. The main points I wanted to clarify are:

  1. a maintainer can merge a PR based on the reviews of other project contributors they trust to competently review and test changes. They do not need to do a detailed review and test each and every PR themselves.
  2. a maintainer must ensure "rough consensus" is reached on the need for and correctness of a PR.
  3. at least one two maintainers must ACK a PR who is not the original author.

Notes to the reviewers

If these changes work for bdk_wallet I'll propose the same change on our other repos. I will also reach out to current maintainers to suggest additional people who they would like to add (and are willing to be) maintainers. Having only two per repo does seem to be enough, the rust-bitcoin team has eight people with maintainer rights.

Changelog notice

  • Add maintainers section to CONTRIBUTING.md file.

Before submitting

@notmandatory notmandatory self-assigned this Jul 29, 2026
@notmandatory notmandatory added the documentation Improvements or additions to documentation label Jul 29, 2026
@notmandatory notmandatory moved this to Needs Review in BDK Wallet Jul 29, 2026
@notmandatory
notmandatory force-pushed the docs/update_contributing branch from 11dfdd2 to ecbda92 Compare July 29, 2026 19:53
@codecov

codecov Bot commented Jul 29, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 81.91%. Comparing base (fc88144) to head (610fc5d).
⚠️ Report is 8 commits behind head on master.

Additional details and impacted files
@@            Coverage Diff             @@
##           master     #521      +/-   ##
==========================================
+ Coverage   81.84%   81.91%   +0.07%     
==========================================
  Files          25       25              
  Lines        6487     6535      +48     
  Branches      296      302       +6     
==========================================
+ Hits         5309     5353      +44     
+ Misses       1078     1075       -3     
- Partials      100      107       +7     
Flag Coverage Δ
rust 81.91% <ø> (+0.07%) ⬆️

Flags with carried forward coverage won't be shown. Click here to find out more.

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@notmandatory
notmandatory force-pushed the docs/update_contributing branch from ecbda92 to 0102c27 Compare July 29, 2026 20:57
Comment thread CONTRIBUTING.md Outdated
@notmandatory
notmandatory force-pushed the docs/update_contributing branch 2 times, most recently from 22f61f6 to 3c4b3c5 Compare August 14, 2026 14:28
@notmandatory notmandatory added this to the Wallet 3.2.0 milestone Aug 14, 2026
@notmandatory

Copy link
Copy Markdown
Member Author

I added this to the next milestone but if it doesn't get adequate review can move to the next one.

@notmandatory

Copy link
Copy Markdown
Member Author

This doesn't need to be in 3.2 release so moved to 4.0, would just like to have it on master branch.

@oleonardolima oleonardolima left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

ACK 3c4b3c5

I left a minor non-blocking nit.

Comment thread CONTRIBUTING.md Outdated
- at least two "accepts"/ACK from a repository maintainer (other than the author).
- no reasonable "rejects"/NACK from anybody who reviewed the code.

See the .github/CODEOWNERS file for a current list of the project maintainers.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
See the .github/CODEOWNERS file for a current list of the project maintainers.
See the [CODEOWNERS](.github/CODEOWNERS) file for a current list of the project maintainers.

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

thanks, fixed

@notmandatory
notmandatory force-pushed the docs/update_contributing branch from 3c4b3c5 to 610fc5d Compare September 4, 2026 20:06
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

documentation Improvements or additions to documentation

Projects

Status: Needs Review

Development

Successfully merging this pull request may close these issues.

3 participants