Skip to content

Fix release process - sign release prs - #134

Merged
fvlankvelt merged 1 commit into
mainfrom
fix-release-process
Oct 2, 2026
Merged

fvlankvelt merged 1 commit into
mainfrom
fix-release-process

Conversation

@fvlankvelt

@fvlankvelt fvlankvelt commented Oct 2, 2026 •

Copy link
Copy Markdown
Contributor

see #131

@changeset-bot

changeset-bot Bot commented Oct 2, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f5c7e03

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
observability Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@fvlankvelt
fvlankvelt merged commit 01f5a03 into main Oct 2, 2026
3 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants