Skip to content

STS 551: allow namespace to already exist - #132

Merged
fvlankvelt merged 1 commit into
mainfrom
sts-551-hosted-namespace
Oct 2, 2026
Merged

fvlankvelt merged 1 commit into
mainfrom
sts-551-hosted-namespace

Conversation

@fvlankvelt

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Oct 2, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d54cadd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
observability Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread manifests/namespace-suse-observability-extension.yaml
@fvlankvelt
fvlankvelt merged commit 7dded19 into main Oct 2, 2026
3 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants