Skip to content

Add support for OpenSUSE Leap 16 - #15111

Open
teacup-on-rockingchair wants to merge 6 commits into
ComplianceAsCode:masterfrom
teacup-on-rockingchair:opensuse_leap16
Open

Add support for OpenSUSE Leap 16#15111
teacup-on-rockingchair wants to merge 6 commits into
ComplianceAsCode:masterfrom
teacup-on-rockingchair:opensuse_leap16

Conversation

@teacup-on-rockingchair

Copy link
Copy Markdown
Contributor

Description:

  • Add definitions and product support for OpenSUSE Leap 16

Rationale:

  • Add OpenSUSE Leap 16 product definition
  • Add support for opensuse leap 16 in applicability checks
  • Add support for the OpenSUSE Leap 16 in testing infrastructure, i.e. product stability Dockerfile definitions etc

Review Hints:

  • Defined as new product as it has too many differences with older versions of OpenSUSE

@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Used by openshift-ci bot. label Sep 10, 2026
@Mab879 Mab879 added this to the 0.1.83 milestone Sep 10, 2026
@teacup-on-rockingchair teacup-on-rockingchair added New Product Issues or pull requests related to new Products. Update Template Issues or pull requests related to Templates updates. labels Sep 13, 2026
@openshift-ci

openshift-ci Bot commented Sep 13, 2026

Copy link
Copy Markdown

@teacup-on-rockingchair: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/4.12-images 1782920 link true /test 4.12-images

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

New Product Issues or pull requests related to new Products. Update Template Issues or pull requests related to Templates updates.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants